Skip to content

Fix hydration mismatches with React.useId - #2566

Merged
Andarist merged 2 commits into
v10from
backport-useid-fix
Nov 26, 2021
Merged

Fix hydration mismatches with React.useId#2566
Andarist merged 2 commits into
v10from
backport-useid-fix

Conversation

@Andarist

Copy link
Copy Markdown
Member

No description provided.

@Andarist
Andarist requested a review from emmatown November 25, 2021 23:12
@changeset-bot

changeset-bot Bot commented Nov 25, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 409b0cf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 3 packages
Name Type
@emotion/core Minor
@emotion/styled Minor
@emotion/styled-base Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment on lines +2 to +4
'@emotion/core': minor
'@emotion/styled': minor
'@emotion/styled-base': minor

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Why minor instead of patch? Don't mind strongly either way, just curious

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I've decided to do the same for the types-related change. Kinda doesn't matter but maybe puts less pressure on people to immediately update their code. At the time being they don't have to do the update - they will only have to do that if they start using React 18 and that has not been even released yet.

Dunno, I just sometimes use minor bumps for some additional context than a "new feature". The problem here is that this might affect some people - for example, the serialized output of jest-emotion might change (like it has changed in our tests on this branch here). So this leaves a little bit of breathing room for people so they can stay on the previous version if they want.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sounds good

Comment thread .changeset/strange-kids-change.md Outdated
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants