Skip to content

[release/11.0.1xx] Source code updates from dotnet/msbuild - #9381

Merged
dotnet-maestro[bot] merged 1 commit into
release/11.0.1xxfrom
darc-release/11.0.1xx-4356a4dc-dd70-4ce8-a577-923740296e7b
Sep 15, 2026
Merged

dotnet-maestro[bot] merged 1 commit into
release/11.0.1xxfrom
darc-release/11.0.1xx-4356a4dc-dd70-4ce8-a577-923740296e7b

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:081bc5e1aa3e8780efe1ff61df274a3aef21d4e7..https://github.com/dotnet/dotnet:darc-release/11.0.1xx-4356a4dc-dd70-4ce8-a577-923740296e7b

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Important

⚠️ No subscription flowing code in the opposite direction is set up.
Code changes made in the VMR's src/ directory will not flow back to the source repository on branch "main".

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team September 15, 2026 17:01

@maestro-approver maestro-approver Bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This pull request contains only source updates from https://github.com/dotnet/msbuild.

@dotnet-maestro
dotnet-maestro Bot merged commit c7f398f into release/11.0.1xx Sep 15, 2026
16 checks passed
@dotnet-maestro
dotnet-maestro Bot deleted the darc-release/11.0.1xx-4356a4dc-dd70-4ce8-a577-923740296e7b branch September 15, 2026 19:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants