Skip to content

[main] Source code updates from dotnet/msbuild - #7292

Merged
lewing merged 3 commits into
mainfrom
darc-main-5fe5d2f9-c92d-4697-aeed-907087c76083
Jun 22, 2026
Merged

lewing merged 3 commits into
mainfrom
darc-main-5fe5d2f9-c92d-4697-aeed-907087c76083

Conversation

@dotnet-maestro

@dotnet-maestro dotnet-maestro Bot commented Jun 19, 2026

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:e775a976a210698020837be79dde8db8ddf218ef..https://github.com/dotnet/dotnet:darc-main-5fe5d2f9-c92d-4697-aeed-907087c76083

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 19, 2026 02:26

@lewing lewing left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Approving green code flow PR. CI is green.

@lewing
lewing enabled auto-merge June 22, 2026 15:40
@lewing
lewing merged commit e1bfbb0 into main Jun 22, 2026
16 checks passed
@lewing
lewing deleted the darc-main-5fe5d2f9-c92d-4697-aeed-907087c76083 branch June 22, 2026 15:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant