Skip to content

[release/10.0.4xx] Source code updates from dotnet/msbuild - #7029

Merged
JanProvaznik merged 1 commit into
release/10.0.4xxfrom
darc-release/10.0.4xx-24ddfd1f-ee70-4f0a-8489-fe01e0a9e481
Jun 4, 2026
Merged

JanProvaznik merged 1 commit into
release/10.0.4xxfrom
darc-release/10.0.4xx-24ddfd1f-ee70-4f0a-8489-fe01e0a9e481

Conversation

@dotnet-maestro

@dotnet-maestro dotnet-maestro Bot commented Jun 4, 2026

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:f69fa2f9d5df4fc8d80fb11bf899617f0019d802..https://github.com/dotnet/dotnet:darc-release/10.0.4xx-24ddfd1f-ee70-4f0a-8489-fe01e0a9e481

@dotnet-maestro

dotnet-maestro Bot commented Jun 4, 2026

Copy link
Copy Markdown
Contributor Author

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 4, 2026 02:10
@JanProvaznik
JanProvaznik merged commit 425f840 into release/10.0.4xx Jun 4, 2026
13 checks passed
@JanProvaznik
JanProvaznik deleted the darc-release/10.0.4xx-24ddfd1f-ee70-4f0a-8489-fe01e0a9e481 branch June 4, 2026 08:42
@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 4, 2026 08:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant