Skip to content

[main] Source code updates from dotnet/msbuild - #6822

Merged
dotnet-maestro[bot] merged 2 commits into
mainfrom
darc-main-f9462287-a8ff-41d7-9a61-932dd4f71304
May 26, 2026
Merged

dotnet-maestro[bot] merged 2 commits into
mainfrom
darc-main-f9462287-a8ff-41d7-9a61-932dd4f71304

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:4ea8215937724cd480a29c30187195b5cbda7486..https://github.com/dotnet/dotnet:darc-main-f9462287-a8ff-41d7-9a61-932dd4f71304

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team May 26, 2026 02:07
@dotnet-maestro
dotnet-maestro Bot merged commit 86edbc3 into main May 26, 2026
16 checks passed
@dotnet-maestro
dotnet-maestro Bot deleted the darc-main-f9462287-a8ff-41d7-9a61-932dd4f71304 branch May 26, 2026 08:08
@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team May 26, 2026 08:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant