Skip to content

[main] Source code updates from dotnet/msbuild - #6760

Merged
OvesN merged 2 commits into
mainfrom
darc-main-e1806fcb-63f2-4a7c-92cc-4efb3dc2aa88
May 22, 2026
Merged

OvesN merged 2 commits into
mainfrom
darc-main-e1806fcb-63f2-4a7c-92cc-4efb3dc2aa88

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:0c1489549759e9282dc07ed27068051006ac35d4..https://github.com/dotnet/dotnet:darc-main-e1806fcb-63f2-4a7c-92cc-4efb3dc2aa88

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team May 22, 2026 02:10
@OvesN
OvesN merged commit f0ca255 into main May 22, 2026
16 checks passed
@OvesN
OvesN deleted the darc-main-e1806fcb-63f2-4a7c-92cc-4efb3dc2aa88 branch May 22, 2026 12:40
@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team May 22, 2026 12:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant