Repository navigation
fix: prevent null prefixed results from React.useId when using client:only directive - #8075
Merged
natemoo-re merged 2 commits intoAug 14, 2023
Conversation
🦋 Changeset detectedLatest commit: 40c1fa5 The changes in this PR will be included in the next version bump. Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
SudoCat
force-pushed
the
feat/react-clientside-identifier-prefix
branch
from
August 14, 2023 15:24
30806ca to
b299846
Compare
SudoCat
force-pushed
the
feat/react-clientside-identifier-prefix
branch
from
August 14, 2023 15:25
b299846 to
c6e8429
Compare
This was causing React components rendered with client:only to be prefixed with null. While not technically causing any issues, it is unintended and could be considered a bug.
SudoCat
force-pushed
the
feat/react-clientside-identifier-prefix
branch
from
August 14, 2023 15:26
c6e8429 to
411b612
Compare
natemoo-re
approved these changes
Aug 14, 2023
natemoo-re
left a comment
Member
There was a problem hiding this comment.
Thanks for the fix, really appreciate the PR!
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
There was a low-impact bug with the unique id generation fix, causing client:only ids to be generated with
null, e.g.:nullr0:The ids generated are actually still unique, and avoid clashes with server-side, but it was flagged by a team member as a possible bug, so I felt an upstream fix was wise.
Changes
nullTesting
Docs
No docs required, no change to user behaviour.