Skip to content

Update React integration changeset - #6698

Merged
bholmesdev merged 1 commit into
mainfrom
react-integration-changeset
Mar 28, 2023
Merged

bholmesdev merged 1 commit into
mainfrom
react-integration-changeset

Conversation

@bholmesdev

Copy link
Copy Markdown
Contributor

Changes

  • Bump React integration to release new README

Testing

Docs

@changeset-bot

changeset-bot Bot commented Mar 28, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b648ec8

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@bholmesdev
bholmesdev merged commit fc71c3f into main Mar 28, 2023
@bholmesdev
bholmesdev deleted the react-integration-changeset branch March 28, 2023 20:38
@astrobot-houston astrobot-houston mentioned this pull request Mar 28, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants