Skip to content

Fix astro add adding lines from extended configurations - #5095

Merged
Princesseuh merged 1 commit into
mainfrom
fix-astro-add-extends
Oct 14, 2022
Merged

Princesseuh merged 1 commit into
mainfrom
fix-astro-add-extends

Conversation

@Princesseuh

Copy link
Copy Markdown
Member

Changes

astro add was mistakenly merging the full extended configuration instead of just the found one, this fix that. Additionally, the react-dom types were missing so I went ahead and added that

Testing

Tested manually

Docs

N/A

@changeset-bot

changeset-bot Bot commented Oct 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 33c37b5

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added pkg: astro Related to the core `astro` package (scope) pkg: react Related to React (scope) pkg: integration Related to any renderer integration (scope) labels Oct 14, 2022
@Princesseuh Princesseuh self-assigned this Oct 14, 2022
@Princesseuh
Princesseuh merged commit ddfbef5 into main Oct 14, 2022
@Princesseuh
Princesseuh deleted the fix-astro-add-extends branch October 14, 2022 18:16
@astrobot-houston astrobot-houston mentioned this pull request Oct 14, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

pkg: astro Related to the core `astro` package (scope) pkg: integration Related to any renderer integration (scope) pkg: react Related to React (scope)

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants