Skip to content

@astrojs/react - Expose Babel config. - #10675

Merged
matthewp merged 2 commits into
withastro:mainfrom
YowaiCoder:main
Apr 4, 2024
Merged

matthewp merged 2 commits into
withastro:mainfrom
YowaiCoder:main

Conversation

@YowaiCoder

Copy link
Copy Markdown

Changes

  • Expose Babel config for @astro/react, as it was expected to be a "very high priority" in this proposal.

Testing

It should not break the current tests.

Docs

It wasn't specifically mentioned before.

@changeset-bot

changeset-bot Bot commented Apr 4, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8611e91

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added pkg: react Related to React (scope) pkg: integration Related to any renderer integration (scope) labels Apr 4, 2024
Comment thread .changeset/popular-mangos-yawn.md Outdated
@@ -0,0 +1,5 @@
---
"@astrojs/react": patch

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

if this is a new feature this should be minor.

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sorry I messed up, my first contribution to OSS. It is closed unintentionally, I've discarded my changes and performed again, should I reopen it or what?

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

no worries, yeah reopening should be fine.

@YowaiCoder YowaiCoder closed this Apr 4, 2024
@YowaiCoder YowaiCoder reopened this Apr 4, 2024
@matthewp
matthewp merged commit 14f1d49 into withastro:main Apr 4, 2024
@astrobot-houston astrobot-houston mentioned this pull request Apr 4, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

pkg: integration Related to any renderer integration (scope) pkg: react Related to React (scope)

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants