fix(remix-dev): flat route fixes and enhancements - #5228
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: d5750c9 The changes in this PR will be included in the next version bump. This PR includes changesets to release 18 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
mcansh
marked this pull request as ready for review
January 24, 2023 17:35
mcansh
force-pushed
the
logan/flat-route-fixes
branch
2 times, most recently
from
January 24, 2023 17:41
50e9842 to
5d77e95
Compare
mcansh
commented
Jan 25, 2023
mcansh
commented
Jan 25, 2023
1 task
This was
linked to
issues
Jan 29, 2023
TODO: fix layout from being flagged as index route Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
useRouteLoaderData("routes/_landing") should work for "routes/_landing.tsx" and "routes/_landing/index.tsx"
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
Signed-off-by: Logan McAnsh <logan@mcan.sh>
mcansh
force-pushed
the
logan/flat-route-fixes
branch
from
January 30, 2023 22:48
8dd6e6a to
06e0367
Compare
Signed-off-by: Logan McAnsh <logan@mcan.sh>
1 task
1 task
This was referenced Feb 6, 2023
Contributor
|
🤖 Hello there, We just published version Thanks! |
Contributor
|
Fixed my issues perfectly! ❤️ |
1 task
|
Same for me. Perfect ! |
This was referenced Feb 7, 2023
TooTallNate
added a commit
to vercel/vercel
that referenced
this pull request
Feb 23, 2023
There were some fixes to thier new v2 filesystem routing that this will help with. Related to remix-run/remix#5322 and remix-run/remix#5228.
kodiakhq Bot
pushed a commit
to vercel/vercel
that referenced
this pull request
Feb 23, 2023
There were some fixes to thier new v2 filesystem routing that this will help with. Related to remix-run/remix#5322 and remix-run/remix#5228.
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Signed-off-by: Logan McAnsh logan@mcan.sh
Closes: #5170
Closes: #5261
Closes: #5303
Closes: #5322
Closes: #5324
Closes: #5327
Testing Strategy:
updated current tests