Skip to content

[release/10.0.1xx] Source code updates from dotnet/msbuild - #8852

Merged
wtgodbe merged 2 commits into
release/10.0.1xxfrom
darc-release/10.0.1xx-7e4cec28-148d-4edc-9e68-52dff3408389
Aug 31, 2026
Merged

wtgodbe merged 2 commits into
release/10.0.1xxfrom
darc-release/10.0.1xx-7e4cec28-148d-4edc-9e68-52dff3408389

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:af2509a16d34876dd2eda43b061c7aa046aaf926..https://github.com/dotnet/dotnet:darc-release/10.0.1xx-7e4cec28-148d-4edc-9e68-52dff3408389

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team August 31, 2026 10:58
@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

This pull request was not automatically approved because the following files did not match the source diff:

  • eng/Versions.props

Automatic approval requires either file changes to match between the repo and the VMR or the destination file to fully match the source file in content
The files listed above met neither condition.

Please manually review these files to ensure their changes are expected before merging this pull request.

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@wtgodbe
wtgodbe merged commit 1fe11c5 into release/10.0.1xx Aug 31, 2026
16 checks passed
@wtgodbe
wtgodbe deleted the darc-release/10.0.1xx-7e4cec28-148d-4edc-9e68-52dff3408389 branch August 31, 2026 19:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant