Skip to content

[main] Source code updates from dotnet/msbuild - #7372

Merged
lewing merged 1 commit into
mainfrom
darc-main-201fe75a-904e-4a64-98d5-935ee8f76177
Jun 25, 2026
Merged

lewing merged 1 commit into
mainfrom
darc-main-201fe75a-904e-4a64-98d5-935ee8f76177

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:33d17672fac4c26656a201774f9894dff51a4e78..https://github.com/dotnet/dotnet:darc-main-201fe75a-904e-4a64-98d5-935ee8f76177

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 25, 2026 02:09

@lewing lewing left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

QB: CI green, forward flow looks clean.

@lewing
lewing enabled auto-merge (squash) June 25, 2026 15:15
@lewing
lewing merged commit 4354817 into main Jun 25, 2026
16 checks passed
@lewing
lewing deleted the darc-main-201fe75a-904e-4a64-98d5-935ee8f76177 branch June 25, 2026 15:15
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant