Skip to content

[release/10.0.4xx] Source code updates from dotnet/msbuild - #7336

Merged
JanProvaznik merged 1 commit into
release/10.0.4xxfrom
darc-release/10.0.4xx-6dc614f2-039b-4e24-986a-c3a00c3a4eff
Jun 24, 2026
Merged

JanProvaznik merged 1 commit into
release/10.0.4xxfrom
darc-release/10.0.4xx-6dc614f2-039b-4e24-986a-c3a00c3a4eff

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:a47a64a9382828e968c3564d11f4e0afe99dc526..https://github.com/dotnet/dotnet:darc-release/10.0.4xx-6dc614f2-039b-4e24-986a-c3a00c3a4eff

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 24, 2026 02:07
@JanProvaznik
JanProvaznik merged commit 706b5c6 into release/10.0.4xx Jun 24, 2026
13 checks passed
@JanProvaznik
JanProvaznik deleted the darc-release/10.0.4xx-6dc614f2-039b-4e24-986a-c3a00c3a4eff branch June 24, 2026 09:42
@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 24, 2026 09:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant