Skip to content

[main] Source code updates from dotnet/msbuild - #3838

Merged
dotnet-maestro[bot] merged 1 commit into
mainfrom
darc-main-9c4df8ad-d8a0-45ba-8488-f09138c8ff5e
Dec 12, 2025
Merged

dotnet-maestro[bot] merged 1 commit into
mainfrom
darc-main-9c4df8ad-d8a0-45ba-8488-f09138c8ff5e

Conversation

@dotnet-maestro

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/msbuild

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/msbuild:2235ffd325c3926d0392fae9dcc2dcb6a623f1eb..https://github.com/dotnet/dotnet:darc-main-9c4df8ad-d8a0-45ba-8488-f09138c8ff5e

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team December 12, 2025 02:03
@dotnet-maestro
dotnet-maestro Bot merged commit fba5fc3 into main Dec 12, 2025
16 checks passed
@dotnet-maestro
dotnet-maestro Bot deleted the darc-main-9c4df8ad-d8a0-45ba-8488-f09138c8ff5e branch December 12, 2025 10:20
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant