Skip to content

Graduate 'DisallowInsecureCSRUsageDefinition' to GA (part 2) - #6963

Merged
cert-manager-prow[bot] merged 1 commit into
cert-manager:masterfrom
inteon:graduate_DisallowInsecureCSRUsageDefinition
Apr 26, 2024
Merged

cert-manager-prow[bot] merged 1 commit into
cert-manager:masterfrom
inteon:graduate_DisallowInsecureCSRUsageDefinition

Conversation

@inteon

@inteon inteon commented Apr 26, 2024

Copy link
Copy Markdown
Member

In #6927, we graduated this feature in the webhook.
However, that PR did not include graduating this feature in the controller and removing the deprecated code.
That is what this PR does.

Kind

/kind cleanup

Release Note

Graduate the 'DisallowInsecureCSRUsageDefinition' feature gate to GA. (part 2)

Signed-off-by: Tim Ramlot <42113979+inteon@users.noreply.github.com>
@cert-manager-prow cert-manager-prow Bot added release-note Denotes a PR that will be considered when it comes time to generate release notes. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. area/testing Issues relating to testing size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Apr 26, 2024

@SgtCoDFish SgtCoDFish left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

Thanks for being diligent with graduating this!

@cert-manager-prow cert-manager-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 26, 2024
@cert-manager-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: SgtCoDFish

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@cert-manager-prow cert-manager-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 26, 2024
@SgtCoDFish SgtCoDFish added this to the 1.15 milestone Apr 26, 2024
@inteon

inteon commented Apr 26, 2024

Copy link
Copy Markdown
Member Author

/retest

@cert-manager-prow
cert-manager-prow Bot merged commit 410b7a6 into cert-manager:master Apr 26, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/testing Issues relating to testing dco-signoff: yes Indicates that all commits in the pull request have the valid DCO sign-off message. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm Indicates that a PR is ready to be merged. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants